Skip to content

refactor(docs): added post pagination to icons - #8437

Open
AndrinAAF wants to merge 8 commits into
mainfrom
6449-pagination-implement-in-icons-search-doc-page
Open

refactor(docs): added post pagination to icons#8437
AndrinAAF wants to merge 8 commits into
mainfrom
6449-pagination-implement-in-icons-search-doc-page

Conversation

@AndrinAAF

Copy link
Copy Markdown
Contributor

📄 Description

Please include a summary of the changes made in this PR.

🚀 Demo

If applicable, please add a screenshot or video to illustrate the changes.


🔮 Design review

  • Design review done
  • No design review needed

🧪 Visual regression tests

  • Visual changes detected and approved (Check this box if VRT fails and changes are intentional)

📝 Checklist

  • ✅ My code follows the style guidelines of this project
  • 🛠️ I have performed a self-review of my own code
  • 📄 I have made corresponding changes to the documentation
  • ⚠️ My changes generate no new warnings or errors
  • 🧪 I have added tests that prove my fix is effective or that my feature works
  • ✔️ New and existing unit tests pass locally with my changes

@AndrinAAF
AndrinAAF requested review from a team as code owners September 8, 2026 09:14
@changeset-bot

changeset-bot Bot commented Sep 8, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ddc2a85

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@swisspost/design-system-documentation Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@swisspost-bot

swisspost-bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

Related Previews

@myrtasakellariou myrtasakellariou linked an issue Sep 8, 2026 that may be closed by this pull request
2 tasks
@AndrinAAF AndrinAAF changed the title design(docs): added post pagination to icons refactor(docs): added post pagination to icons Sep 9, 2026
@sonarqubecloud

sonarqubecloud Bot commented Sep 9, 2026

Copy link
Copy Markdown

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Implement the new pagination component in the icons search docs page

2 participants