Skip to content

Update deps to fix auto-update error - #2352

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/run-update-deps
Sep 14, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
dsimansk:pr/run-update-deps

Conversation

@dsimansk

@dsimansk dsimansk commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Fixes #

Proposed Changes

  • Update deps to fix auto-update error

Chore cleanup of argument list too long on the auto-updates pipeline runs.

Generated by:

./hack/update-deps.sh --upgrade --release 1.24

https://github.com/knative-extensions/knobots/actions/runs/34796297419/job/103830042138

/cc @gauron99 @kahirokunn

Release Note

NONE

@knative-prow knative-prow Bot added the size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. label Sep 14, 2026
@knative-prow

knative-prow Bot commented Sep 14, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dsimansk

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Sep 14, 2026
@gauron99

Copy link
Copy Markdown

/lgtm

@knative-prow knative-prow Bot added the lgtm Indicates that a PR is ready to be merged. label Sep 14, 2026
@codecov

codecov Bot commented Sep 14, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 65.95%. Comparing base (bd7bb89) to head (f8e4ea4).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main    #2352   +/-   ##
=======================================
  Coverage   65.95%   65.95%           
=======================================
  Files          56       56           
  Lines        2682     2682           
=======================================
  Hits         1769     1769           
  Misses        777      777           
  Partials      136      136           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@knative-prow
knative-prow Bot merged commit 074eebe into knative:main Sep 14, 2026
26 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. lgtm Indicates that a PR is ready to be merged. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants