Skip to content

[main] Upgrade to latest dependencies - #2256

Merged
knative-prow[bot] merged 1 commit into
knative:mainfrom
knative-automation:auto-updates/update-deps-main
Jul 29, 2026
Merged

knative-prow[bot] merged 1 commit into
knative:mainfrom
knative-automation:auto-updates/update-deps-main

Conversation

@knative-automation

Copy link
Copy Markdown
Contributor

Cron -dprotaso

/cc knative/client-wg-leads
/assign knative/client-wg-leads

Produced by: knative-extensions/knobots/actions/update-deps

bumping knative.dev/serving 333e58c...0b59f4a:
  > 0b59f4a Fix flaky TestMinScaleAnnotationChange by waiting for scale convergence (# 16668)
  > d8212f5 Update net-istio nightly (# 16672)
  > 108bfba Bump the github-actions group across 1 directory with 3 updates (# 16655)

Signed-off-by: Knative Automation <automation@knative.team>
@knative-prow
knative-prow Bot requested a review from a team July 29, 2026 01:49
@knative-prow

knative-prow Bot commented Jul 29, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: knative-automation
Once this PR has been reviewed and has the lgtm label, please assign dprotaso for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@knative-prow knative-prow Bot added the size/S Denotes a PR that changes 10-29 lines, ignoring generated files. label Jul 29, 2026
@codecov

codecov Bot commented Jul 29, 2026

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 71.88%. Comparing base (11438fb) to head (e2796d5).
⚠️ Report is 1 commits behind head on main.

Additional details and impacted files
@@           Coverage Diff           @@
##             main    #2256   +/-   ##
=======================================
  Coverage   71.88%   71.88%           
=======================================
  Files         233      233           
  Lines       13535    13535           
=======================================
  Hits         9730     9730           
  Misses       2999     2999           
  Partials      806      806           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.

@knative-prow
knative-prow Bot merged commit 0e1aa1e into knative:main Jul 29, 2026
19 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/S Denotes a PR that changes 10-29 lines, ignoring generated files. skip-review

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant